Conversation
|
Hi @YahiaBadr. Thanks for your PR. I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with Tip We noticed you've done this a few times! Consider joining the org to skip this step and gain Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: YahiaBadr The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
What's the diff here? What is the default here? |
https://github.com/kubernetes/kops/blob/master/tests/e2e/scenarios/scalability/run-test.sh#L136 c5.2xlarge (aws machine) |
|
Ok, seems like the defaulting is AWS specific, can we update the run-test.sh to have a separate sensible default for GCE? |
Sure, done kubernetes/kops#18612 |
|
No need for the PR as kubernetes/kops#18612 is merged |
No description provided.