Sitelet https://github.com/kubernetes/kops/pull/18612
Skip to content

Adding suitable control plane machine for gce - #18612

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
YahiaBadr:controle-plane-machine
Jul 20, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
YahiaBadr:controle-plane-machine

Conversation

@YahiaBadr

Copy link
Copy Markdown
Contributor

No description provided.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 20, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

Hi @YahiaBadr. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow kubernetes-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Jul 20, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from dims and serathius July 20, 2026 14:55
@YahiaBadr YahiaBadr changed the title Adding suitable controle plane machine for gce Adding suitable control plane machine for gce Jul 20, 2026
Comment thread tests/e2e/scenarios/scalability/run-test.sh
@YahiaBadr
YahiaBadr force-pushed the controle-plane-machine branch from c6dfdf4 to a52a068 Compare July 20, 2026 15:13
@YahiaBadr
YahiaBadr force-pushed the controle-plane-machine branch from a52a068 to 26e5a14 Compare July 20, 2026 15:14
@serathius

Copy link
Copy Markdown
Member

Guessing c3-standard-8 should have similar performance of c5.2xlarge based on fact they have 8 cores, at least based on Google search.

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 20, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 20, 2026
@serathius

Copy link
Copy Markdown
Member

/ok-to-test

@kubernetes-prow kubernetes-prow Bot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jul 20, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit c525f93 into kubernetes:master Jul 20, 2026
20 checks passed
@YahiaBadr
YahiaBadr deleted the controle-plane-machine branch July 20, 2026 17:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants