Sitelet https://github.com/kubernetes/test-infra/pull/37024
Skip to content

kops: don't set AZURE_STORAGE_ACCOUNT - #37024

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:kops-gossip-upgrades
May 13, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
hakman:kops-gossip-upgrades

Conversation

@hakman

@hakman hakman commented May 13, 2026

Copy link
Copy Markdown
Member

No need for this after kubernetes/kops#18260.

/cc @upodroid @ameukam

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and upodroid May 13, 2026 09:47
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/config Issues or PRs related to code in /config size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. area/jobs sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels May 13, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 13, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, upodroid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit befc6a5 into kubernetes:master May 13, 2026
6 checks passed
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Updated the job-config configmap in namespace default at cluster test-infra-trusted using the following files:

  • key kops-presets.yaml using file config/jobs/kubernetes/kops/kops-presets.yaml
Details

In response to this:

No need for this after kubernetes/kops#18260.

/cc @upodroid @ameukam

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants