Sitelet https://github.com/kubernetes/kops/pull/18685
Skip to content

gce: open the etcd events metrics port in the node-to-master firewall - #18685

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:gce-firewall-etcd-events-metrics
Aug 14, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:gce-firewall-etcd-events-metrics

Conversation

@Jefftree

@Jefftree Jefftree commented Aug 13, 2026 •

Copy link
Copy Markdown
Member

open the etcd events metrics port in the node-to-master firewall

xref: #18568

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Aug 13, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from hakman and olemarkus August 13, 2026 16:20
@kubernetes-prow kubernetes-prow Bot added area/documentation area/provider/gcp Issues or PRs related to gcp provider labels Aug 13, 2026
@Jefftree

Copy link
Copy Markdown
Member Author

/assign @hakman

@Jefftree

Copy link
Copy Markdown
Member Author

/cc @serathius

@serathius

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 13, 2026
@hakman

hakman commented Aug 14, 2026

Copy link
Copy Markdown
Member

@Jefftree this looks fine but we may need to change the metrics ports to not overlap with other things already managed by kOps. Would that be ok for you?

@Jefftree

Copy link
Copy Markdown
Member Author

@Jefftree this looks fine but we may need to change the metrics ports to not overlap with other things already managed by kOps. Would that be ok for you?

of course, what do you recommend?

@hakman

hakman commented Aug 14, 2026

Copy link
Copy Markdown
Member

@Jefftree this looks fine but we may need to change the metrics ports to not overlap with other things already managed by kOps. Would that be ok for you?

of course, what do you recommend?

Thanks! Let's merge this and let me do some research in the mean time. Will get back to you on Slack.
/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 14, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 20efe0b into kubernetes:master Aug 14, 2026
27 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation area/provider/gcp Issues or PRs related to gcp provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants