Sitelet https://github.com/kubernetes/kops/pull/18568
Skip to content

scalability: expose events-etcd metrics endpoint - #18568

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:expose-events-etcd-metrics
Jul 10, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:expose-events-etcd-metrics

Conversation

@Jefftree

@Jefftree Jefftree commented Jul 9, 2026 •

Copy link
Copy Markdown
Member

Expose events etcd metrics, we've only scraped main etcd before. Skipped 2383 as that's used for leases.

@kubernetes-prow
kubernetes-prow Bot requested review from dims and hakman July 9, 2026 21:26
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 9, 2026
@Jefftree

Jefftree commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/assign @serathius

@serathius

Copy link
Copy Markdown
Member

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added lgtm "Looks good to me", indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jul 10, 2026
@hakman

hakman commented Jul 10, 2026

Copy link
Copy Markdown
Member

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Jefftree

Copy link
Copy Markdown
Member Author

/retest

@kubernetes-prow
kubernetes-prow Bot merged commit e078f00 into kubernetes:master Jul 10, 2026
28 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants