Sitelet https://github.com/kubernetes/kops/pull/18674
Skip to content

Release 1.36.2 - #18674

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.36from
hakman:release-1.36.2
Aug 8, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.36from
hakman:release-1.36.2

Conversation

@hakman

@hakman hakman commented Aug 8, 2026

Copy link
Copy Markdown
Member

@hakman hakman added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Aug 8, 2026
@kubernetes-prow
kubernetes-prow Bot requested a review from ameukam August 8, 2026 11:18
@kubernetes-prow kubernetes-prow Bot added this to the v1.36 milestone Aug 8, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from justinsb and rifelpet August 8, 2026 11:18
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Aug 8, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 8, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 8, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit f336152 into kubernetes:release-1.36 Aug 8, 2026
17 checks passed
social4hyq pushed a commit to social4hyq/homebrew-core that referenced this pull request Sep 20, 2026
kops 1.36.2

Created-by: HarmonybrewBot
Commit-by: HarmonybrewBot
Merged-by: HarmonybrewBot
Description: Created by `brew bump`

---

Created with `brew bump-formula-pr`.<details>
  <summary>release notes</summary>
  <pre>## What's Changed
* Automated cherry pick of #18636: azure: support terraform provider azurerm v5 by @hakman in kubernetes/kops#18637
* Automated cherry pick of #18638: Update hcloud-go/v2 to latest version (v2.47.0) by @hakman in kubernetes/kops#18639
* Automated cherry pick of #18641: rolling-update: restore 5s delay between pod eviction retries by @hakman in kubernetes/kops#18643
* Automated cherry pick of #18642: aws: do not set the primary network interface to "Unmanaged=yes" by @hakman in kubernetes/kops#18644
* Automated cherry pick of #18645: etcd-manager: compute kops-utils-cp image at use time by @hakman in kubernetes/kops#18646
* Automated cherry pick of #18649: aws: Default to AWS CCM v1.36.1 for all k8s versions by @rifelpet in kubernetes/kops#18652
* Automated cherry pick of #18657: fix(channels): only treat known ready conditions as health signals by @hakman in kubernetes/kops#18659
* Automated cherry pick of #18655: Generate EC2NodeClass block device mapping for root volume by @hakman in kubernetes/kops#18673
* Release 1.36.2 by @hakman in kubernetes/kops#18674


**Full Changelog**: https://github.com/kubernetes/kops/compare/v1.36.1...v1.36.2</pre>
  <p>View the full release notes at <a href="/sitelet?url=https%3A%2F%2Fgithub.com%2Fkubernetes%2Fkops%2Fpull%2F%253Ca%2520href%3D"https://github.com/kubernetes/kops/releases/tag/v1.36.2">https://github.com/kubernetes/kops/releases/tag/v1.36.2</a>.</p">https://github.com/kubernetes/kops/releases/tag/v1.36.2">https://github.com/kubernetes/kops/releases/tag/v1.36.2</a>.</p>
</details>
<hr>

See merge request: Harmonybrew/homebrew-core!16183
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants