Sitelet https://github.com/kubernetes/kops/pull/18634
Skip to content

Release 1.36.1 - #18634

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.36from
hakman:release_1.36.1
Jul 28, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.36from
hakman:release_1.36.1

Conversation

@hakman

@hakman hakman commented Jul 28, 2026

Copy link
Copy Markdown
Member

@hakman hakman added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Jul 28, 2026
@kubernetes-prow
kubernetes-prow Bot requested a review from ameukam July 28, 2026 09:51
@kubernetes-prow kubernetes-prow Bot added this to the v1.36 milestone Jul 28, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from rifelpet and upodroid July 28, 2026 09:51
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jul 28, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 28, 2026
@hakman

hakman commented Jul 28, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, upodroid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 28, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 9a00448 into kubernetes:release-1.36 Jul 28, 2026
18 checks passed
social4hyq pushed a commit to social4hyq/homebrew-core that referenced this pull request Sep 20, 2026
kops 1.36.1

Created-by: HarmonybrewBot
Commit-by: HarmonybrewBot
Merged-by: HarmonybrewBot
Description: Created by `brew bump`

---

Created with `brew bump-formula-pr`.<details>
  <summary>release notes</summary>
  <pre>## What's Changed
* Automated cherry pick of #18600: nodeup: don't mutate the cluster spec when clearing authenticator config by @rifelpet in kubernetes/kops#18602
* Automated cherry pick of #18631: hetzner: Ignore WellKnownServices when comparing load balancers by @hakman in kubernetes/kops#18633
* Release 1.36.1 by @hakman in kubernetes/kops#18634


**Full Changelog**: https://github.com/kubernetes/kops/compare/v1.36.0...v1.36.1</pre>
  <p>View the full release notes at <a href="/sitelet?url=https%3A%2F%2Fgithub.com%2Fkubernetes%2Fkops%2Fpull%2F%253Ca%2520href%3D"https://github.com/kubernetes/kops/releases/tag/v1.36.1">https://github.com/kubernetes/kops/releases/tag/v1.36.1</a>.</p">https://github.com/kubernetes/kops/releases/tag/v1.36.1">https://github.com/kubernetes/kops/releases/tag/v1.36.1</a>.</p>
</details>
<hr>

See merge request: Harmonybrew/homebrew-core!15315
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants