Sitelet https://github.com/kubernetes/kops/pull/18548
Skip to content

Automated cherry pick of #18546: Update Go to 1.26.5 and bump golang.org/x modules - #18548

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18546-upstream-release-1.35
Jul 7, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18546-upstream-release-1.35

Conversation

@hakman

@hakman hakman commented Jul 7, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18546 on release-1.35.

#18546: Update Go to 1.26.5 and bump golang.org/x modules

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?

NONE

@kubernetes-prow kubernetes-prow Bot added this to the v1.35 milestone Jul 7, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 7, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab July 7, 2026 20:39
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 7, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 7, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 659f2f5 into kubernetes:release-1.35 Jul 7, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants