Sitelet https://github.com/kubernetes/kops/pull/18546
Skip to content

Update Go to 1.26.5 and bump golang.org/x modules - #18546

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:bump-go-1.26.5
Jul 7, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:bump-go-1.26.5

Conversation

@hakman

@hakman hakman commented Jul 7, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 7, 2026 20:25
@kubernetes-prow kubernetes-prow Bot added size/L Denotes a PR that changes 100-499 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jul 7, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 7, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 7, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 2916a97 into kubernetes:master Jul 7, 2026
28 checks passed
kubernetes-prow Bot added a commit that referenced this pull request Jul 7, 2026
…-upstream-release-1.36

Automated cherry pick of #18546: Update Go to 1.26.5 and bump golang.org/x modules
kubernetes-prow Bot added a commit that referenced this pull request Jul 7, 2026
…-upstream-release-1.35

Automated cherry pick of #18546: Update Go to 1.26.5 and bump golang.org/x modules
@hakman
hakman deleted the bump-go-1.26.5 branch July 14, 2026 19:45
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants