Sitelet https://github.com/kubernetes/kops/pull/18457
Skip to content

scaltest: Default node dump timeout to 5m in scalability run-test.sh - #18457

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Jefftree:node-dump-timeout-run-test
Jun 8, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
Jefftree:node-dump-timeout-run-test

Conversation

@Jefftree

@Jefftree Jefftree commented Jun 8, 2026

Copy link
Copy Markdown
Member

Follow up to #18443. Instead of remembering to set this for every scalability job on test-infra side, just default them to 5m.

/cc @serathius @hakman

@k8s-ci-robot
k8s-ci-robot requested review from hakman and serathius June 8, 2026 13:52
@k8s-ci-robot k8s-ci-robot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jun 8, 2026
@serathius

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 8, 2026
@hakman hakman changed the title Default node dump timeout to 5m in scalability run-test.sh scaltest: Default node dump timeout to 5m in scalability run-test.sh Jun 8, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 8, 2026
@serathius

Copy link
Copy Markdown
Member

/retest

@k8s-ci-robot
k8s-ci-robot merged commit 551a385 into kubernetes:master Jun 8, 2026
28 checks passed
@Jefftree
Jefftree deleted the node-dump-timeout-run-test branch June 26, 2026 03:36
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants