Sitelet https://github.com/kubernetes/kops/pull/18422
Skip to content

chore: upgrade containerd to v2.3.1 - #18422

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:containerd-2.3.1
May 28, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
hakman:containerd-2.3.1

Conversation

@hakman

@hakman hakman commented May 28, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet May 28, 2026 05:15
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels May 28, 2026
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member Author

/retest

@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 28, 2026
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member Author

Unrelated flakes:
/override pull-kops-e2e-cni-kuberouter
/override pull-kops-e2e-azure-cni-calico

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-azure-cni-calico, pull-kops-e2e-cni-kuberouter

Details

In response to this:

Unrelated flakes:
/override pull-kops-e2e-cni-kuberouter
/override pull-kops-e2e-azure-cni-calico

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

k8s-ci-robot added a commit that referenced this pull request May 28, 2026
…18422-upstream-release-1.35

Automated cherry pick of #18364: chore: upgrade containerd to v2.3.0
#18422: chore: upgrade containerd to v2.3.1
@ameukam

ameukam commented May 28, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 28, 2026
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-k8s-aws-amazonvpc

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-amazonvpc

Details

In response to this:

/override pull-kops-e2e-k8s-aws-amazonvpc

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot
k8s-ci-robot merged commit 24864df into kubernetes:master May 28, 2026
40 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants