Sitelet https://github.com/kubernetes/kops/pull/18364
Skip to content

chore: upgrade containerd to v2.3.0 - #18364

Merged
k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
hakman:containerd-2.3.0
May 19, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
hakman:containerd-2.3.0

Conversation

@hakman

@hakman hakman commented May 18, 2026

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from ameukam and rifelpet May 18, 2026 12:57
@k8s-ci-robot k8s-ci-robot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels May 18, 2026
@hakman

hakman commented May 18, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-e2e-cni-calico

@hakman

hakman commented May 18, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-cni-kuberouter

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-cni-kuberouter

Details

In response to this:

/override pull-kops-e2e-cni-kuberouter

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ameukam

ameukam commented May 18, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 18, 2026
@hakman

hakman commented May 18, 2026

Copy link
Copy Markdown
Member Author

/test all

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

PR needs rebase.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 18, 2026
@hakman
hakman force-pushed the containerd-2.3.0 branch from 134b327 to 4d1618f Compare May 18, 2026 19:19
@k8s-ci-robot k8s-ci-robot removed lgtm "Looks good to me", indicates that a PR is ready to be merged. needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. labels May 18, 2026
@hakman

hakman commented May 19, 2026

Copy link
Copy Markdown
Member Author

/retest

@hakman

hakman commented May 19, 2026

Copy link
Copy Markdown
Member Author

/test all

@hakman

hakman commented May 19, 2026

Copy link
Copy Markdown
Member Author

/approve

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 19, 2026
@hakman

hakman commented May 19, 2026

Copy link
Copy Markdown
Member Author

Small flakes:

[It] [sig-network] Services should implement service.kubernetes.io/headless

/override pull-kops-e2e-cni-kuberouter

[It] External Storage [Driver: disk.csi.azure.com] [Testpattern: Dynamic PV (ext3)] volumes should store data

/override pull-kops-e2e-azure-cni-calico

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-azure-cni-calico, pull-kops-e2e-cni-kuberouter

Details

In response to this:

Small flakes:

[It] [sig-network] Services should implement service.kubernetes.io/headless

/override pull-kops-e2e-cni-kuberouter

[It] External Storage [Driver: disk.csi.azure.com] [Testpattern: Dynamic PV (ext3)] volumes should store data

/override pull-kops-e2e-azure-cni-calico

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ameukam

ameukam commented May 19, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 19, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 6e75cb6 into kubernetes:master May 19, 2026
41 of 42 checks passed
@hakman hakman changed the title Upgrade containerd to v2.3.0 chore: upgrade containerd to v2.3.0 May 28, 2026
k8s-ci-robot added a commit that referenced this pull request May 28, 2026
…18422-upstream-release-1.35

Automated cherry pick of #18364: chore: upgrade containerd to v2.3.0
#18422: chore: upgrade containerd to v2.3.1
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants