Sitelet https://github.com/kubernetes/kops/pull/18421
Skip to content

Upgrade kube-router to v2.10.0 - #18421

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
rifelpet:kuberouter-2100
May 28, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
rifelpet:kuberouter-2100

Conversation

@rifelpet

Copy link
Copy Markdown
Member

@k8s-ci-robot
k8s-ci-robot requested review from hakman and zetaab May 28, 2026 02:18
@k8s-ci-robot k8s-ci-robot added area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels May 28, 2026
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member

/test pull-kops-e2e-cni-kuberouter

1 similar comment
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member

/test pull-kops-e2e-cni-kuberouter

Comment thread upup/models/cloudup/resources/addons/networking.kuberouter/k8s-1.12.yaml.template Outdated
Comment thread upup/models/cloudup/resources/addons/networking.kuberouter/k8s-1.12.yaml.template Outdated
Comment thread upup/models/cloudup/resources/addons/networking.kuberouter/k8s-1.12.yaml.template Outdated
Comment thread upup/models/cloudup/resources/addons/networking.kuberouter/k8s-1.12.yaml.template Outdated
@hakman

hakman commented May 28, 2026

Copy link
Copy Markdown
Member

/test pull-kops-e2e-cni-kuberouter

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label May 28, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 28, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 4d09962 into kubernetes:master May 28, 2026
27 of 29 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants