Sitelet https://github.com/kubernetes/kops/pull/18218
Skip to content

Automated cherry pick of #18110: Update Cluster Autoscaler to 1.35.0 - #18218

Merged
k8s-ci-robot merged 2 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18110-upstream-release-1.34
Apr 19, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:release-1.34from
hakman:automated-cherry-pick-of-#18110-upstream-release-1.34

Conversation

@hakman

@hakman hakman commented Apr 19, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18110 on release-1.34.

#18110: Update Cluster Autoscaler to 1.35.0

For details on the cherry pick process, see the cherry pick requests page.


ameukam added 2 commits April 19, 2026 05:03
Bump CAS default version and previous versions

Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@k8s-ci-robot k8s-ci-robot added this to the v1.34 milestone Apr 19, 2026
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Apr 19, 2026
@hakman

hakman commented Apr 19, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet @ameukam

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Apr 19, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 19, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 9d238ba into kubernetes:release-1.34 Apr 19, 2026
18 checks passed
@hakman
hakman deleted the automated-cherry-pick-of-#18110-upstream-release-1.34 branch May 22, 2026 09:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants