Sitelet https://github.com/kubernetes/kops/pull/18110
Skip to content

Update Cluster Autoscaler to 1.35.0 - #18110

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
ameukam:bump-cluster-autoscaler-1-35
Mar 23, 2026
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
ameukam:bump-cluster-autoscaler-1-35

Conversation

@ameukam

@ameukam ameukam commented Mar 22, 2026

Copy link
Copy Markdown
Member

Bump CAS default version and previous versions

Bump CAS default version and previous versions

Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Mar 22, 2026
@ameukam

ameukam commented Mar 22, 2026

Copy link
Copy Markdown
Member Author

cc @hakman @justinsb @rifelpet

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 22, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 22, 2026
@hakman

hakman commented Mar 22, 2026

Copy link
Copy Markdown
Member

@ameukam please run hack/update_expected.sh.
/hold

@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Mar 22, 2026
Signed-off-by: Arnaud Meukam <ameukam@gmail.com>
@k8s-ci-robot k8s-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Mar 23, 2026
@hakman

hakman commented Mar 23, 2026

Copy link
Copy Markdown
Member

/lgtm
/unhold

@k8s-ci-robot k8s-ci-robot added lgtm "Looks good to me", indicates that a PR is ready to be merged. and removed do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. labels Mar 23, 2026
@k8s-ci-robot
k8s-ci-robot merged commit ae04e59 into kubernetes:master Mar 23, 2026
27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 23, 2026
k8s-ci-robot added a commit that referenced this pull request Apr 19, 2026
…-upstream-release-1.34

Automated cherry pick of #18110: Update Cluster Autoscaler to 1.35.0
k8s-ci-robot added a commit that referenced this pull request Apr 19, 2026
…-upstream-release-1.35

Automated cherry pick of #18110: Update Cluster Autoscaler to 1.35.0
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants