Sitelet https://github.com/kubernetes/kops/pull/18098
Skip to content

Enable E2E external CSI testing on GCP - #18098

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rifelpet:gcp-pd-e2e
Mar 19, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rifelpet:gcp-pd-e2e

Conversation

@rifelpet

Copy link
Copy Markdown
Member

We were only setting the --storage.testdriver flag for AWS. Now we set it for GCP as well. kops enables GCP's PD CSI driver by default so we shouldn't need to adjust anything on the cluster spec side.

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Mar 19, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@rifelpet: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-kops-kubernetes-e2e-ubuntu-gce-build 14dbafd link false /test pull-kops-kubernetes-e2e-ubuntu-gce-build

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Let's give it a try! 🚀

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Mar 19, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 19, 2026
@k8s-ci-robot
k8s-ci-robot merged commit d25b7f8 into kubernetes:master Mar 19, 2026
26 of 27 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Mar 19, 2026
@hdp617

hdp617 commented Mar 19, 2026

Copy link
Copy Markdown

I'm seeing failures in cloud-provider-gcp test similar to the failing pull-kops-kubernetes-e2e-ubuntu-gce-build above. Do you have any recommendation? Thanks!

  invalid value "/home/prow/go/src/k8s.io/kubernetes/tests/e2e/csi-manifests/gcp-pd/driver.yaml" for flag -storage.testdriver: open /home/prow/go/src/k8s.io/kubernetes/tests/e2e/csi-manifests/gcp-pd/driver.yaml: no such file or directory

@rifelpet

Copy link
Copy Markdown
Member Author

Interesting, it seems specific to presubmits because the periodics are passing:

https://testgrid.k8s.io/kops-gce#kops-gce-cni-calico

I'm guessing os.Getwd() isn't a safe assumption here. I'll open a fix.

@rifelpet

Copy link
Copy Markdown
Member Author

proposed fix here: #18102

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants