scaletest: Use sleep instead of wait for large clusters - #18065
Merged
Merged
Conversation
Member
Author
|
/test pull-kops-gce-master-scale-performance-100 |
upodroid
approved these changes
Mar 13, 2026
Member
|
/approve |
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hakman, upodroid The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
BenTheElder
reviewed
Mar 13, 2026
| --set=spec.image="${INSTANCE_IMAGE:-ubuntu-os-cloud/ubuntu-2404-noble-amd64-v20251001}" | ||
| kops update cluster --yes | ||
| kops validate cluster --wait 10m | ||
| # TODO: Replaced with `kops validate instancegroup --wait 10m` when it's fixed for 5k node clusters. |
Merged
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixing https://prow.k8s.io/view/gs/kubernetes-ci-logs/logs/ci-kubernetes-e2e-gce-scale-performance-5000/2032502308796895232
Investigation in https://kubernetes.slack.com/archives/C09QZTRH7/p1773429346033779?thread_ts=1773428126.202359&cid=C09QZTRH7
Looks like
kops validate cluster --wait 10mtimedout even though node was ready for over 8 minutes./assign @hakman @upodroid