Sitelet https://github.com/kubernetes/kops/pull/18010
Skip to content

fix: avoid panic if subnet ID is nil - #18010

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:nlb_npe
Feb 23, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
justinsb:nlb_npe

Conversation

@justinsb

Copy link
Copy Markdown
Member

Not addressing root-cause, but avoid a panic if the subnet ID is nil.

Not addressing root-cause, but avoid a panic if the subnet ID is nil.
@k8s-ci-robot k8s-ci-robot added the area/provider/aws Issues or PRs related to aws provider label Feb 23, 2026
@k8s-ci-robot
k8s-ci-robot requested a review from zetaab February 23, 2026 14:11
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Feb 23, 2026
@justinsb

Copy link
Copy Markdown
Member Author

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Feb 23, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 23, 2026
@k8s-ci-robot
k8s-ci-robot merged commit 775adb2 into kubernetes:master Feb 23, 2026
26 checks passed
@k8s-ci-robot k8s-ci-robot added this to the v1.36 milestone Feb 23, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/aws Issues or PRs related to aws provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants