tests/ai-conformance: create first tests for AI Conformance - #18009
Merged
Merged
Conversation
Member
Author
|
/test pull-kops-ai-conformance |
ameukam
reviewed
Feb 23, 2026
justinsb
force-pushed
the
aiconform_custom_tests
branch
from
February 23, 2026 13:09
75c8dea to
1ffd92b
Compare
Member
Author
|
/test pull-kops-ai-conformance |
Mostly working on the harness here, so we can produce output in the correct format. Co-authored-by: Arnaud M. <ameukam@gmail.com>
justinsb
force-pushed
the
aiconform_custom_tests
branch
from
February 23, 2026 13:42
1ffd92b to
e29eb9a
Compare
Member
Author
|
Fixed the shellcheck /test pull-kops-ai-conformance |
Member
Author
|
/test pull-kops-e2e-k8s-aws-amazonvpc Partial fix for that is in #18010 |
Member
Author
|
Be sure to peek at the artifacts: https://gcsweb.k8s.io/gcs/kubernetes-ci-logs/pr-logs/pull/kops/18009/pull-kops-ai-conformance/2025929245527642112/artifacts/ai-conformance/ :-) |
Member
Member
|
/approve |
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hakman The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Member
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Mostly working on the harness here, so we can produce output in the correct format.