Sitelet https://github.com/wevm/references/pull/491
Skip to content
This repository was archived by the owner on Sep 6, 2023. It is now read-only.

chore: Update WalletConnect dependencies - #491

Merged
tmm merged 1 commit into
mainfrom
chore/update-wc-dependencies
Aug 29, 2023
Merged

tmm merged 1 commit into
mainfrom
chore/update-wc-dependencies

Conversation

@xzilja

@xzilja xzilja commented Aug 29, 2023

Copy link
Copy Markdown
Contributor

Description

Updated @walletconnect dependencies

Additional Information

Your ENS/address:

@changeset-bot

changeset-bot Bot commented Aug 29, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2c16810

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@wagmi/connectors Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@xzilja
xzilja requested review from jxom and tmm August 29, 2023 12:32
@github-actions

Copy link
Copy Markdown
Contributor

Size Change: 0 B

Total Size: 58.3 kB

ℹ️ View Unchanged
Filename Size Change
packages/chains/dist 10.9 kB 0 B
packages/chains/dist/index.d.ts 11.4 kB 0 B
packages/chains/dist/index.js 10.9 kB 0 B
packages/chains/dist/index.mjs 1.36 kB 0 B
packages/connectors/dist 198 B 0 B
packages/connectors/dist/base-98051e7c.d.ts 784 B 0 B
packages/connectors/dist/chunk-OQILYQDO.js 2.73 kB 0 B
packages/connectors/dist/chunk-QYMCVNHT.js 292 B 0 B
packages/connectors/dist/chunk-YZRVUC6Y.js 729 B 0 B
packages/connectors/dist/chunk-ZCAPXGBX.js 1.8 kB 0 B
packages/connectors/dist/coinbaseWallet.d.ts 313 B 0 B
packages/connectors/dist/coinbaseWallet.js 165 B 0 B
packages/connectors/dist/index.d.ts 970 B 0 B
packages/connectors/dist/index.js 119 B 0 B
packages/connectors/dist/injected.d.ts 760 B 0 B
packages/connectors/dist/injected.js 1.99 kB 0 B
packages/connectors/dist/ledger.d.ts 518 B 0 B
packages/connectors/dist/ledger.js 1.3 kB 0 B
packages/connectors/dist/metaMask.d.ts 0 B 0 B 🆕
packages/connectors/dist/metaMask.js 808 B 0 B
packages/connectors/dist/mock 1.62 kB 0 B
packages/connectors/dist/mock/index.d.ts 739 B 0 B
packages/connectors/dist/mock/index.js 1.17 kB 0 B
packages/connectors/dist/safe.d.ts 1.65 kB 0 B
packages/connectors/dist/safe.js 2.72 kB 0 B
packages/connectors/dist/walletConnect.d.ts 508 B 0 B
packages/connectors/dist/walletConnect.js 1.76 kB 0 B
packages/connectors/dist/walletConnectLegacy.d.ts 0 B 0 B 🆕
packages/connectors/dist/walletConnectLegacy.js 0 B 0 B 🆕

compressed-size-action

@glitch-txs

Copy link
Copy Markdown
Contributor

LGTM 😆

@tmm
tmm merged commit 0306383 into main Aug 29, 2023
@tmm
tmm deleted the chore/update-wc-dependencies branch August 29, 2023 15:07
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants