Sitelet https://github.com/openstack-k8s-operators/s2i-openstack-containers/pull/255
Skip to content

Do not use the "function" keyword - #255

Open
CyrilRoelandteNovance wants to merge 1 commit into
openstack-k8s-operators:mainfrom
CyrilRoelandteNovance:do-not-use-function-keyword
Open

CyrilRoelandteNovance wants to merge 1 commit into
openstack-k8s-operators:mainfrom
CyrilRoelandteNovance:do-not-use-function-keyword

Conversation

@CyrilRoelandteNovance

Copy link
Copy Markdown

The "function" keyword is non standard. In most shell scripts in this
repository, it is not used. Let us be consistent and never use it.

@elfiesmelfie elfiesmelfie left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks okay to me, but please solicit feedback from horizon and tempest contributers.

}


# This function ensures all arguments are handled properly:

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Your regular expression needs tuning :)

Please restore this line for future grok-ability

Copy link
Copy Markdown
Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yes, there were two lines like these :) My sed skills escaped me for a second. Nice catch!

The "function" keyword is non standard. In most shell scripts in this
repository, it is not used. Let us be consistent and never use it.
@CyrilRoelandteNovance
CyrilRoelandteNovance force-pushed the do-not-use-function-keyword branch from 138e7cf to ba4e6ed Compare October 6, 2026 18:54
@openshift-ci

openshift-ci Bot commented Oct 6, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please ask for approval from elfiesmelfie. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@centosinfra-prod-github-app

Copy link
Copy Markdown

Build failed (check pipeline). Post recheck (without leading slash)
to rerun all jobs. Make sure the failure cause has been resolved before
you rerun jobs.

https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/ccbd13ae1c8949fa9c5c484b659e754f

❌ horizon-s2i-test POST_FAILURE in 15m 19s
✔️ s2i-openstack-containers-molecule SUCCESS in 2m 03s
✔️ s2i-openstack-container-content-provider SUCCESS in 4h 43m 28s
✔️ s2i-openstack-container-consumer-smoke SUCCESS in 6m 03s
✔️ s2i-openstack-deploy-validation SUCCESS in 1h 27m 37s

@HanzJas

HanzJas commented Oct 7, 2026

Copy link
Copy Markdown
Contributor

recheck

@centosinfra-prod-github-app

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants