Repository navigation
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Build failed (check pipeline). Post ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 01s |
7583276 to
da8854b
Compare
|
Zuul encountered a syntax error while parsing its found undefined alias |
da8854b to
bff0011
Compare
|
Build failed (check pipeline). Post ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 03s |
6ad4058 to
5319035
Compare
|
Build failed (check pipeline). Post ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 06s |
5319035 to
e83f60b
Compare
|
Unable to freeze job graph: Job watcher-s2i-tempest depends on s2i-openstack-container-consumer-smoke which was not run. |
e83f60b to
3c171e1
Compare
|
Build failed (check pipeline). Post ✔️ s2i-openstack-containers-molecule SUCCESS in 2m 02s |
3c171e1 to
1a41254
Compare
|
Build succeeded (check pipeline). ✔️ s2i-openstack-containers-molecule SUCCESS in 1m 59s |
1a41254 to
3474b81
Compare
|
Build succeeded (check pipeline). ✔️ s2i-openstack-containers-molecule SUCCESS in 5m 22s |
3474b81 to
0eaaf07
Compare
|
Build failed (check pipeline). Post ❌ ironic-s2i-hotstack-sno-1-bm FAILURE in 1h 36m 50s |
|
recheck |
|
Build failed (check pipeline). Post
|
|
recheck Mirror issues? |
|
Build succeeded (check pipeline). ✔️ ironic-s2i-hotstack-sno-1-bm SUCCESS in 1h 37m 56s |
Include project template ironic-operator-s2i-component-jobs from ironic-operator. Signed-off-by: Harald Jensas <hjensas@redhat.com>
0eaaf07 to
a9a537d
Compare
elfiesmelfie
left a comment
There was a problem hiding this comment.
/lgtm
Good to merge once CI passes.
|
Build succeeded (check pipeline). ✔️ ironic-s2i-hotstack-sno-1-bm SUCCESS in 1h 39m 48s |
Include project template
ironic-operator-s2i-component-jobsfromironic-operator