Sitelet https://github.com/limitless-angular/limitless-angular/pull/13
Skip to content

feat(sanity): unknown components and reduce template logic - #13

Merged
osnoser1 merged 15 commits into
mainfrom
feature/portable-text-complexity
Mar 4, 2025
Merged

osnoser1 merged 15 commits into
mainfrom
feature/portable-text-complexity

Conversation

@osnoser1

@osnoser1 osnoser1 commented Mar 1, 2025 •

Copy link
Copy Markdown
Member

PR Checklist

What is the new behavior?

All this changes are related only to portable-text.

  • Expose properties to set custom unknown components.
  • Reduce template logic.
  • Refactor code to make it more maintainable.

Does this PR introduce a breaking change?

  • Yes
  • No

Other information

[Optional] What gif best describes this PR or how it makes you feel?

@osnoser1 osnoser1 added the enhancement New feature or request label Mar 1, 2025
@netlify

netlify Bot commented Mar 1, 2025 •

Copy link
Copy Markdown

✅ Deploy Preview for limitless-angular-sanity-example ready!

Name Link
🔨 Latest commit ab3705f
🔍 Latest deploy log https://app.netlify.com/sites/limitless-angular-sanity-example/deploys/67c5f3082554cc00087658a5
😎 Deploy Preview https://deploy-preview-13--limitless-angular-sanity-example.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@nx-cloud

nx-cloud Bot commented Mar 1, 2025 •

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit ab3705f.

Command Status Duration Result
nx affected -t build test ✅ Succeeded 1m 36s View ↗

☁️ Nx Cloud last updated this comment at 2025-03-03 18:23:09 UTC

@pkg-pr-new

pkg-pr-new Bot commented Mar 1, 2025 •

Copy link
Copy Markdown

Open in Stackblitz

npm i https://pkg.pr.new/@limitless-angular/sanity@13

commit: ab3705f

@osnoser1
osnoser1 force-pushed the feature/portable-text-complexity branch 2 times, most recently from ad6fffb to 725e684 Compare March 2, 2025 22:53
@osnoser1
osnoser1 force-pushed the feature/portable-text-complexity branch from 725e684 to 0f745cf Compare March 2, 2025 23:03
@osnoser1
osnoser1 force-pushed the feature/portable-text-complexity branch 2 times, most recently from 1ea4150 to 0f2eef7 Compare March 3, 2025 04:05
@osnoser1
osnoser1 force-pushed the feature/portable-text-complexity branch from 0f2eef7 to cd61ba1 Compare March 3, 2025 04:29
@osnoser1
osnoser1 merged commit 26b6b3b into main Mar 4, 2025
@osnoser1
osnoser1 deleted the feature/portable-text-complexity branch March 4, 2025 00:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

enhancement New feature or request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant