Sitelet https://github.com/kubernetes/test-infra/pull/37417
Skip to content

Drop ETCD_VERSION 3.7.0-rc.0 override from sig-scalability scale jobs - #37417

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:drop-etcd-3.7-override
Jul 9, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
Jefftree:drop-etcd-3.7-override

Conversation

@Jefftree

@Jefftree Jefftree commented Jul 9, 2026

Copy link
Copy Markdown
Member

kubernetes/kops#18557 defaults all tests to etcd 3.7, so this override is unnecessary and technically breaking since it replaces the released 3.7 with the 3.7 rc.

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. area/config Issues or PRs related to code in /config area/jobs sig/scalability Categorizes an issue or PR as relevant to SIG Scalability. labels Jul 9, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from mm4tt and wojtek-t July 9, 2026 02:09
@kubernetes-prow kubernetes-prow Bot added the sig/testing Categorizes an issue or PR as relevant to SIG Testing. label Jul 9, 2026
@Jefftree

Jefftree commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/assign @serathius

@serathius

Copy link
Copy Markdown
Member

/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 9, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Jefftree, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 9, 2026
@serathius

serathius commented Jul 9, 2026 •

Copy link
Copy Markdown
Member

/hold

Would it better for future release to keep the environment variable targetting on v3.7.0 so it's clear how to change it for v3.8? Scalability tests should not care about kops default etcd version.

@kubernetes-prow kubernetes-prow Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 9, 2026
@Jefftree

Jefftree commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/hold

Would it better for future release to keep the environment variable targetting on v3.7.0 so it's clear how to change it for v3.8? Scalability tests should not care about kops default etcd version.

Personally I think it's quite reasonable even for a new contributor to figure out where to add the etcd version just by grepping through the flags. I would prefer dropping it because we copy around configs a lot, and it's easy to have a stale config with a flag that hasn't been updated in a while.

@serathius

Copy link
Copy Markdown
Member

/unhold

@kubernetes-prow kubernetes-prow Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 9, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 2cb524e into kubernetes:master Jul 9, 2026
6 checks passed
@kubernetes-prow

Copy link
Copy Markdown
Contributor

@Jefftree: Updated the job-config configmap in namespace default at cluster test-infra-trusted using the following files:

  • key sig-scalability-experimental-presubmit-jobs.yaml using file config/jobs/kubernetes/sig-scalability/sig-scalability-experimental-presubmit-jobs.yaml
  • key sig-scalability-periodic-ec2.yaml using file config/jobs/kubernetes/sig-scalability/sig-scalability-periodic-ec2.yaml
  • key sig-scalability-periodic-jobs.yaml using file config/jobs/kubernetes/sig-scalability/sig-scalability-periodic-jobs.yaml
  • key sig-scalability-presubmit-jobs.yaml using file config/jobs/kubernetes/sig-scalability/sig-scalability-presubmit-jobs.yaml
Details

In response to this:

kubernetes/kops#18557 defaults all tests to etcd 3.7, so this override is unnecessary and technically breaking since it replaces the released 3.7 with the 3.7 rc.

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@Jefftree
Jefftree deleted the drop-etcd-3.7-override branch July 9, 2026 21:43
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/scalability Categorizes an issue or PR as relevant to SIG Scalability. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants