Sitelet https://github.com/kubernetes/test-infra/pull/37250
Skip to content

Stop pinning arm64 grid jobs to eu-west-1a - #37250

Merged
k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
rifelpet:euw1a
Jun 13, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
rifelpet:euw1a

Conversation

@rifelpet

Copy link
Copy Markdown
Member

With kubernetes/kops#18463 merged we can start removing hardcoded zones from jobs. This starts with the grid jobs.

/cc @hakman

@k8s-ci-robot
k8s-ci-robot requested a review from hakman June 13, 2026 15:27
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. labels Jun 13, 2026
@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 13, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 19341bd into kubernetes:master Jun 13, 2026
6 checks passed
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@rifelpet: Updated the job-config configmap in namespace default at cluster test-infra-trusted using the following files:

  • key kops-periodics-grid.yaml using file config/jobs/kubernetes/kops/kops-periodics-grid.yaml
Details

In response to this:

With kubernetes/kops#18463 merged we can start removing hardcoded zones from jobs. This starts with the grid jobs.

/cc @hakman

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants