Sitelet https://github.com/kubernetes/test-infra/pull/36996
Skip to content

Kops - Skip Ubuntu 26.04 on Kops 1.35 - #36996

Closed
rifelpet wants to merge 2 commits into
kubernetes:masterfrom
rifelpet:2604-135
Closed

rifelpet wants to merge 2 commits into
kubernetes:masterfrom
rifelpet:2604-135

Conversation

@rifelpet

Copy link
Copy Markdown
Member

Support for 26.04 was added in Kops 1.36, so testing with 1.35 fails:

https://storage.googleapis.com/kubernetes-ci-logs/logs/e2e-kops-grid-cilium-etcd-u2604-k35-ko35/2052853460642041856/artifacts/52.78.206.57/journal.log

May 08 21:11:30.898712 i-046a9838693041cd7 cloud-init[516]: W0508 21:11:30.898658 1184 main.go:133] got error running nodeup (will retry in 30s): error determining OS distribution: unsupported distro "ubuntu-26.04"

Removes 76 failing jobs:

diff --git a/config/jobs/kubernetes/kops/kops-periodics-grid.yaml b/config/jobs/kubernetes/kops/kops-periodics-grid.yaml
index b64adab909..eec5053a5e 100644
--- a/config/jobs/kubernetes/kops/kops-periodics-grid.yaml
+++ b/config/jobs/kubernetes/kops/kops-periodics-grid.yaml
@@ -1,5 +1,5 @@
 # Test jobs generated by build_jobs.py (do not manually edit)
-# 2967 jobs, total of 18138 runs per week
+# 2891 jobs, total of 18062 runs per week
...

/cc @hakman

@k8s-ci-robot
k8s-ci-robot requested a review from hakman May 10, 2026 12:44
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels May 10, 2026
@hakman

hakman commented May 10, 2026

Copy link
Copy Markdown
Member

@rifelpet should we maybe cherry pick to 1.35 instead?

@rifelpet

Copy link
Copy Markdown
Member Author

I'm fine with that too

@hakman

hakman commented May 10, 2026

Copy link
Copy Markdown
Member

If you can do it, I can approve

@rifelpet

Copy link
Copy Markdown
Member Author

kubernetes/kops#18306

@rifelpet rifelpet closed this May 10, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants