Sitelet https://github.com/kubernetes/test-infra/pull/36344
Skip to content

drop AL2 from kops - #36344

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
upodroid:drop-al2-from-kops
Jan 30, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
upodroid:drop-al2-from-kops

Conversation

@upodroid

@upodroid upodroid commented Jan 30, 2026 •

Copy link
Copy Markdown
Member

/cc @hakman @dims

Dropping AL2 as its going EoL soon and it doesn't support cgroupsv2 and it has a high failure rate

https://storage.googleapis.com/k8s-metrics/failures-latest.html?include=amzn2&sort=days&asc=0

@k8s-ci-robot
k8s-ci-robot requested review from dims and hakman January 30, 2026 17:50
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/config Issues or PRs related to code in /config area/jobs size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Jan 30, 2026
skip_jobs = []

image = "gcr.io/k8s-staging-test-infra/kubekins-e2e:v20260127-f10a7ebcce-master"
image = "us-central1-docker.pkg.dev/k8s-staging-test-infra/images/kubekins-e2e:v20260127-c1affcc8de-master"

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@dims

dims commented Jan 30, 2026

Copy link
Copy Markdown
Member

/approve

@dims

dims commented Jan 30, 2026

Copy link
Copy Markdown
Member

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 30, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dims, upodroid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit ac36e0c into kubernetes:master Jan 30, 2026
6 checks passed
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

@upodroid: Updated the job-config configmap in namespace default at cluster test-infra-trusted using the following files:

  • key kops-periodics-conformance.yaml using file config/jobs/kubernetes/kops/kops-periodics-conformance.yaml
  • key kops-periodics-distros.yaml using file config/jobs/kubernetes/kops/kops-periodics-distros.yaml
  • key kops-periodics-grid.yaml using file config/jobs/kubernetes/kops/kops-periodics-grid.yaml
  • key kops-periodics-misc2.yaml using file config/jobs/kubernetes/kops/kops-periodics-misc2.yaml
  • key kops-periodics-network-plugins.yaml using file config/jobs/kubernetes/kops/kops-periodics-network-plugins.yaml
  • key kops-periodics-nftables.yaml using file config/jobs/kubernetes/kops/kops-periodics-nftables.yaml
  • key kops-periodics-pipeline.yaml using file config/jobs/kubernetes/kops/kops-periodics-pipeline.yaml
  • key kops-periodics-upgrades.yaml using file config/jobs/kubernetes/kops/kops-periodics-upgrades.yaml
  • key kops-periodics-versions.yaml using file config/jobs/kubernetes/kops/kops-periodics-versions.yaml
  • key kops-presubmits-distros.yaml using file config/jobs/kubernetes/kops/kops-presubmits-distros.yaml
  • key kops-presubmits-e2e.yaml using file config/jobs/kubernetes/kops/kops-presubmits-e2e.yaml
  • key kops-presubmits-network-plugins.yaml using file config/jobs/kubernetes/kops/kops-presubmits-network-plugins.yaml
  • key kops-presubmits-scale.yaml using file config/jobs/kubernetes/kops/kops-presubmits-scale.yaml
Details

In response to this:

/cc @hakman @dims

Dropping AL2 as its going EoL soon and it doesn't support cgroupsv2 and it has a high failure rate

https://storage.googleapis.com/k8s-metrics/failures-latest.html?include=amzn2&sort=days&asc=0

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/config Issues or PRs related to code in /config area/jobs area/testgrid cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. sig/cluster-lifecycle Categorizes an issue or PR as relevant to SIG Cluster Lifecycle. sig/testing Categorizes an issue or PR as relevant to SIG Testing. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants