Sitelet https://github.com/kubernetes/kops/pull/18704
Skip to content

Update Flannel to v0.28.9 - #18704

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:flannel-v0.28.9
Aug 17, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:flannel-v0.28.9

Conversation

@hakman

@hakman hakman commented Aug 17, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet August 17, 2026 04:34
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/addons size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Aug 17, 2026
@ameukam

ameukam commented Aug 17, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 17, 2026
@hakman

hakman commented Aug 17, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 17, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit cf30825 into kubernetes:master Aug 17, 2026
30 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants