feat(akamai): Add volume task and extend cloud interface & mock for volume mgmt - #18700
Merged
kubernetes-prow[bot] merged 1 commit intoAug 17, 2026
Merged
Conversation
Member
Author
|
/retest |
hakman
reviewed
Aug 17, 2026
Comment on lines
+127
to
+137
| name := fi.ValueOf(expected.Name) | ||
| if name == "" { | ||
| return fi.RequiredField("Name") | ||
| } | ||
|
|
||
| if expected.Region == nil { | ||
| return fi.RequiredField("Region") | ||
| } | ||
| if expected.SizeGB == nil { | ||
| return fi.RequiredField("SizeGB") | ||
| } |
Member
There was a problem hiding this comment.
Are these still needed after CheckChanges()
Member
Author
There was a problem hiding this comment.
Good call, thanks for catching that!
Removed. 🙏🏼
hakman
reviewed
Aug 17, 2026
| if expected.SizeGB == nil { | ||
| return fi.RequiredField("SizeGB") | ||
| } | ||
| if fi.ValueOf(expected.SizeGB) <= 0 { |
Member
There was a problem hiding this comment.
The minimum size is 10 GB.
Suggested change
| if fi.ValueOf(expected.SizeGB) <= 0 { | |
| if fi.ValueOf(expected.SizeGB) <= 10 { |
…olume mgmt Signed-off-by: Moshe Vayner <moshe@vayner.me>
moshevayner
force-pushed
the
akamai-volumes
branch
from
August 17, 2026 04:19
900c8d5 to
bec5cee
Compare
hakman
approved these changes
Aug 17, 2026
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hakman The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Signed-off-by: Moshe Vayner moshe@vayner.me
What this PR does / why we need it:
Which issue(s) this PR fixes (optional, in
fixes #<issue number>(, fixes #<issue_number>, ...)format, will close the issue(s) when PR gets merged):Special notes for your reviewer:
This is the first phase of adding support for Akamai volumes support.
Future PRs will include adding the VolumeModelBuilder, actual Akamai/Linode client implementations etc.
I'll make sure to keep these PRs small so that' they are easier to review.
I left some TODO items for myself so that I don't forget adding those.