refactor(linode): build SSH keys with instances - #18694
Merged
kubernetes-prow[bot] merged 1 commit intoAug 15, 2026
Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes-prow[bot] merged 1 commit into
Conversation
hakman
reviewed
Aug 15, 2026
| return nil | ||
| } | ||
|
|
||
| func (b *InstanceModelBuilder) buildSSHKeyTasks(c *fi.CloudupModelBuilderContext) ([]*linodetasks.SSHKey, error) { |
Member
There was a problem hiding this comment.
This returns a slice that is always nil or length 1. Returning *linodetasks.SSHKey would be slightly simpler and more inline with other providers.
Signed-off-by: Moshe Vayner <moshe@vayner.me>
moshevayner
force-pushed
the
linode-sshkey-to-instance-model-builder
branch
from
August 15, 2026 06:40
4ab6fd3 to
8436087
Compare
hakman
approved these changes
Aug 15, 2026
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hakman The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Signed-off-by: Moshe Vayner moshe@vayner.me
What this PR does / why we need it:
Which issue(s) this PR fixes (optional, in
fixes #<issue number>(, fixes #<issue_number>, ...)format, will close the issue(s) when PR gets merged):Special notes for your reviewer:
This PR is addressing the suggested follow up item from #18622
It mainly moves the SSH Model Builder logic into Instance Model Builder.
/cc @hakman