Sitelet https://github.com/kubernetes/kops/pull/18692
Skip to content

chore(networking): bump aws-cni to version 1.23.0 - #18692

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:aws-cni-1.23.0
Aug 15, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:aws-cni-1.23.0

Conversation

@moshevayner

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Special notes for your reviewer:
https://github.com/aws/amazon-vpc-cni-k8s/releases/tag/v1.23.0

Signed-off-by: Moshe Vayner <moshe@vayner.me>
Signed-off-by: Moshe Vayner <moshe@vayner.me>
@kubernetes-prow kubernetes-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 14, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab August 14, 2026 18:36
@kubernetes-prow kubernetes-prow Bot added area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Aug 14, 2026

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks @moshevayner! 🙂

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 15, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 15, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 4f5ca86 into kubernetes:master Aug 15, 2026
30 checks passed
@moshevayner
moshevayner deleted the aws-cni-1.23.0 branch August 15, 2026 04:09
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants