Sitelet https://github.com/kubernetes/kops/pull/18667
Skip to content

Register completion func for kubernetes-feature-gates flag - #18667

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:fix-feature-gates-completion
Aug 6, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:fix-feature-gates-completion

Conversation

@hakman

@hakman hakman commented Aug 6, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and upodroid August 6, 2026 07:16
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Aug 6, 2026
@hakman

hakman commented Aug 6, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 6, 2026
@hakman

hakman commented Aug 6, 2026

Copy link
Copy Markdown
Member Author

/test all
/cc @rifelpet

@kubernetes-prow
kubernetes-prow Bot requested a review from rifelpet August 6, 2026 10:05
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 6, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow
kubernetes-prow Bot merged commit 9dbe32f into kubernetes:master Aug 6, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants