Sitelet https://github.com/kubernetes/kops/pull/18665
Skip to content

Do not upload stale images. - #18665

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
cheftako:minor
Aug 5, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
cheftako:minor

Conversation

@cheftako

@cheftako cheftako commented Aug 4, 2026 •

Copy link
Copy Markdown
Member

This makes no difference on a clean clone.
However if you have been using the clone on previous releases, then you will likely have some older/stale builds. These old stale builds will then be uploaded to your storage. Fixing this is a minor optimization.
It also prevents you(me) accidentily using an older build because we didn't fix our KOPS_BASE_URL path.

This makes no difference on a clean clone.
However is you have been using the clone on previous releases,
then you will likely have some older/stale builds.
These old stale builds will then be uploaded to your storage.
Fixing this is a minor optimization.
It also prevents you(me) accidentily using an older build because
we didn't fix our KOPS_BASE_URL path.
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Aug 4, 2026
@cheftako

cheftako commented Aug 4, 2026

Copy link
Copy Markdown
Member Author

/assign @hakman

@hakman

hakman commented Aug 5, 2026

Copy link
Copy Markdown
Member

Looks great, let's ship it! 😁
/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 5, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 5, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 2fa0b72 into kubernetes:master Aug 5, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants