Sitelet https://github.com/kubernetes/kops/pull/18647
Skip to content

chore(channels): Promote alpha to stable channel - #18647

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:promote-18621
Jul 30, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:promote-18621

Conversation

@moshevayner

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Special notes for your reviewer:
Promotes changes from #18621 into stable channel

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 30, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from hakman and olemarkus July 30, 2026 13:56
Signed-off-by: Moshe Vayner <moshe@vayner.me>
Signed-off-by: Moshe Vayner <moshe@vayner.me>

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks @moshevayner!

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 30, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 30, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 56b7b62 into kubernetes:master Jul 30, 2026
26 checks passed
@moshevayner
moshevayner deleted the promote-18621 branch July 30, 2026 15:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/channels cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants