Sitelet https://github.com/kubernetes/kops/pull/18645
Skip to content

etcd-manager: compute kops-utils-cp image at use time - #18645

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:etcdmanager-utils-cp-image
Jul 30, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:etcdmanager-utils-cp-image

Conversation

@hakman

@hakman hakman commented Jul 30, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 30, 2026 09:39
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Jul 30, 2026
@hakman hakman changed the title etcdmanager: compute kops-utils-cp image at use time etcd-manager: compute kops-utils-cp image at use time Jul 30, 2026
@justinsb

Copy link
Copy Markdown
Member

Good call

/approve
/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 30, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 30, 2026
kubernetes-prow Bot added a commit that referenced this pull request Jul 30, 2026
…-upstream-release-1.36

Automated cherry pick of #18645: etcd-manager: compute kops-utils-cp image at use time
@kubernetes-prow
kubernetes-prow Bot merged commit 1aa87d4 into kubernetes:master Jul 30, 2026
27 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants