Sitelet https://github.com/kubernetes/kops/pull/18638
Skip to content

Update hcloud-go/v2 to latest version (v2.47.0) - #18638

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
krillerdk:hetzner-datacenter-removed-from-api
Jul 29, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
krillerdk:hetzner-datacenter-removed-from-api

Conversation

@krillerdk

Copy link
Copy Markdown
Contributor

The Datacenter field is no longer returned by the Hetzner API (https://docs.hetzner.cloud/changelog#2026-07-01-removing-datacenters)
This means that Datacenter.Location is always nil. Server.Datacenter.Location is now available as Server.Location, but this requires at least hcloud-go/v2 v2.33.0.

The nil value caused all servers to be immediately marked as "needs-update" after they were spun up during "rolling-update cluster".

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Jul 29, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

Hi @krillerdk. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow
kubernetes-prow Bot requested a review from hakman July 29, 2026 10:43
@kubernetes-prow kubernetes-prow Bot added the area/provider/hetzner Issues or PRs related to Hetzner provider label Jul 29, 2026
@kubernetes-prow
kubernetes-prow Bot requested a review from zetaab July 29, 2026 10:43
@hakman

hakman commented Jul 29, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@kubernetes-prow kubernetes-prow Bot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jul 29, 2026
@hakman

hakman commented Jul 29, 2026

Copy link
Copy Markdown
Member

@krillerdk do you mind squashing the commits?

@krillerdk

Copy link
Copy Markdown
Contributor Author

Sure, I will squash them. I just wanted to see that I hadn't missed anything else in terms of failing test cases before I did, so I can do it in one go.

Datacenter is no longer returned by the Hetzner API, so Datacenter.Location is always nil.
Server.Datacenter.Location is now available as Server.Location, but this requires at least hcloud-go/v2 v2.33.0.
@krillerdk
krillerdk force-pushed the hetzner-datacenter-removed-from-api branch from d32bf7d to 5e23610 Compare July 29, 2026 12:51
@krillerdk

Copy link
Copy Markdown
Contributor Author

Squashed.

@hakman hakman left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks @krillerdk!

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 29, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 29, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 4c2516f into kubernetes:master Jul 29, 2026
27 checks passed
kubernetes-prow Bot added a commit that referenced this pull request Jul 29, 2026
…-upstream-release-1.36

Automated cherry pick of #18638: Update hcloud-go/v2 to latest version (v2.47.0)
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/hetzner Issues or PRs related to Hetzner provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants