Sitelet https://github.com/kubernetes/kops/pull/18635
Skip to content

Enable nri mounting on kindnet in kops - #18635

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
michaelasp:updateKindnetNRI
Jul 29, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
michaelasp:updateKindnetNRI

Conversation

@michaelasp

Copy link
Copy Markdown
Contributor

Kindnet allows for the use of nri to enable fast local insertions of pods rather than waiting on pod informers. Enable by mounting the NRI socket into kindnet.

@michaelasp

Copy link
Copy Markdown
Contributor Author

/cc @aojea

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 28, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from aojea, olemarkus and zetaab July 28, 2026 14:15
@michaelasp
michaelasp force-pushed the updateKindnetNRI branch 2 times, most recently from 42e1e6f to 3a22adc Compare July 28, 2026 14:31
@kubernetes-prow kubernetes-prow Bot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. and removed size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 28, 2026
name: var-lib-kindnet
- mountPath: /var/run/nri
name: nri-plugin
- mountPath: /var/run/netns

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

this is a fall back for some previous containerd versions that didn't expose the Pod IPS on shutdown

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

OK, let's remove 🙂

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The netns is a fallback, should I remove that part? Don't mind either way, but might be safer to mount.

@aojea aojea Jul 28, 2026 •

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

if people use containerd < 2.1 the policies will not work fine, is safer to have since 2.0.x is still supported https://github.com/containerd/containerd/releases#release-v2.0.11 and this is exactly how we have it in the official repo

https://github.com/kubernetes-sigs/kindnet/blob/e27de3edfcd69cbcad9fdd50729b2a48c8b40cca/install-kindnet.yaml#L144-L149

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

The netns is a fallback, should I remove that part? Don't mind either way, but might be safer to mount.

I would be happier without the fallback. Thanks!

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

fair

@aojea

aojea commented Jul 28, 2026

Copy link
Copy Markdown
Member

I'm going to release a new kindnet version with the kube-network-policies fix kubernetes-sigs/kube-network-policies#389 so we can just also update the default kindnet version in a separate commit

cc: @hakman

@aojea

aojea commented Jul 28, 2026

Copy link
Copy Markdown
Member

1.0.1 kindnet in its way to promotion https://github.com/kubernetes/k8s.io/pull/9777/changes , minor bugfix for kube-network policies an dependencies updates

@aojea

aojea commented Jul 28, 2026

Copy link
Copy Markdown
Member

@michaelasp we can add another commit to bump the default version to v1.0.1

c.Version = "v1.0.0"

@aojea

aojea commented Jul 28, 2026

Copy link
Copy Markdown
Member

/lgtm

modulo if we want to bump the kindnet default version

@kubernetes-prow kubernetes-prow Bot added lgtm "Looks good to me", indicates that a PR is ready to be merged. and removed lgtm "Looks good to me", indicates that a PR is ready to be merged. labels Jul 28, 2026
@michaelasp

Copy link
Copy Markdown
Contributor Author

kops/pkg/model/components/kindnet.go

Bumped!

@aojea

aojea commented Jul 28, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 28, 2026
@hakman

hakman commented Jul 29, 2026

Copy link
Copy Markdown
Member

Thank you both 🙂
/lgtm
/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 29, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 3962666 into kubernetes:master Jul 29, 2026
33 checks passed
@aojea aojea mentioned this pull request Sep 28, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants