Sitelet https://github.com/kubernetes/kops/pull/18630
Skip to content

aws: allow shared subnets to omit the zone - #18630

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:aws-subnet-zone-optional
Jul 28, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:aws-subnet-zone-optional

Conversation

@hakman

@hakman hakman commented Jul 27, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 27, 2026 06:48
@kubernetes-prow kubernetes-prow Bot added area/provider/aws Issues or PRs related to aws provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 27, 2026
@hakman
hakman force-pushed the aws-subnet-zone-optional branch from 8d60110 to 90e0b45 Compare July 27, 2026 06:52
@ameukam

ameukam commented Jul 27, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 27, 2026
@hakman

hakman commented Jul 28, 2026

Copy link
Copy Markdown
Member Author

/test all

@hakman

hakman commented Jul 28, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 28, 2026
@hakman

hakman commented Jul 28, 2026

Copy link
Copy Markdown
Member Author

/override pull-kops-e2e-k8s-aws-amazonvpc

@kubernetes-prow

Copy link
Copy Markdown
Contributor

@hakman: Overrode contexts on behalf of hakman: pull-kops-e2e-k8s-aws-amazonvpc

Details

In response to this:

/override pull-kops-e2e-k8s-aws-amazonvpc

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kubernetes-prow
kubernetes-prow Bot merged commit 87ff06d into kubernetes:master Jul 28, 2026
27 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/provider/aws Issues or PRs related to aws provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

AWS: Cluster requires designating subnets by both ID and containing AZ

2 participants