Sitelet https://github.com/kubernetes/kops/pull/18615
Skip to content

chore(networking): bump aws-cni to v1.22.3 - #18615

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:aws-vpc-cni-1.22.3
Jul 21, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
moshevayner:aws-vpc-cni-1.22.3

Conversation

@moshevayner

@moshevayner moshevayner commented Jul 21, 2026 •

Copy link
Copy Markdown
Member

What this PR does / why we need it:

Which issue(s) this PR fixes (optional, in fixes #<issue number>(, fixes #<issue_number>, ...) format, will close the issue(s) when PR gets merged):

Special notes for your reviewer:
https://github.com/aws/amazon-vpc-cni-k8s/releases/tag/v1.22.3
Full change log from current version:
aws/amazon-vpc-cni-k8s@v1.21.2...v1.22.3

Signed-off-by: Moshe Vayner <moshe@vayner.me>
Signed-off-by: Moshe Vayner <moshe@vayner.me>
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jul 21, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from hakman and olemarkus July 21, 2026 16:03
@hakman

hakman commented Jul 21, 2026

Copy link
Copy Markdown
Member

@moshevayner are you sure the CRD is being removed?

@moshevayner

Copy link
Copy Markdown
Member Author

@moshevayner are you sure the CRD is being removed?

@hakman yup!
Double checked in the full changelog as well to be sure:

aws/amazon-vpc-cni-k8s@v1.21.2...v1.22.3

@hakman

hakman commented Jul 21, 2026

Copy link
Copy Markdown
Member

Thanks for confirming, weird...
/lgtm
/approve

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 21, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 21, 2026
@moshevayner

Copy link
Copy Markdown
Member Author

@moshevayner are you sure the CRD is being removed?

@hakman yup! Double checked in the full changelog as well to be sure:

aws/amazon-vpc-cni-k8s@v1.21.2...v1.22.3

@hakman looks like this is where they removed it:
aws/amazon-vpc-cni-k8s#3625
Apparently this CRD is installed via the control plane and not as an add on, which is why they removed it from the chart.

@hakman

hakman commented Jul 21, 2026 •

Copy link
Copy Markdown
Member

Apparently this CRD is installed via the control plane and not as an add on, which is why they removed it from the chart.

Yes, but this is not EKS, but seems we don't support the policy engine at all, so we're fine.

The AWS VPC CNI implementation of network policies may be enabled in self-managed clusters. This requires the VPC CNI agent, the Network Policy Controller, and Network Policy Node Agent.

@kubernetes-prow
kubernetes-prow Bot merged commit 36e26d8 into kubernetes:master Jul 21, 2026
29 of 30 checks passed
@moshevayner
moshevayner deleted the aws-vpc-cni-1.22.3 branch July 21, 2026 17:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants