Sitelet https://github.com/kubernetes/kops/pull/18599
Skip to content

docs: update 1.36 release notes post-release - #18599

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:release-notes-1.36-post-release
Jul 16, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:release-notes-1.36-post-release

Conversation

@hakman

@hakman hakman commented Jul 16, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@kubernetes-prow kubernetes-prow Bot added area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 16, 2026
@hakman

hakman commented Jul 16, 2026

Copy link
Copy Markdown
Member Author

/approve
/cc @upodroid

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 16, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 16, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman, rifelpet, upodroid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow
kubernetes-prow Bot merged commit 1ab52d5 into kubernetes:master Jul 16, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants