Sitelet https://github.com/kubernetes/kops/pull/18593
Skip to content

Release 1.35.2 - #18593

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.35from
hakman:release_1.35.2
Jul 15, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:release-1.35from
hakman:release_1.35.2

Conversation

@hakman

@hakman hakman commented Jul 14, 2026

Copy link
Copy Markdown
Member

No description provided.

@hakman hakman added the tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. label Jul 14, 2026
@kubernetes-prow kubernetes-prow Bot added this to the v1.35 milestone Jul 14, 2026
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Jul 14, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab July 14, 2026 20:09
@hakman

hakman commented Jul 14, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet @ameukam

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 14, 2026 20:09
@ameukam

ameukam commented Jul 15, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 15, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 15, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit c89e135 into kubernetes:release-1.35 Jul 15, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants