Sitelet https://github.com/kubernetes/kops/pull/18592
Skip to content

docs: update 1.36 release notes - #18592

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:release-notes-1.36-update
Jul 15, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:release-notes-1.36-update

Conversation

@hakman

@hakman hakman commented Jul 14, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 14, 2026 20:06
@kubernetes-prow kubernetes-prow Bot added area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Jul 14, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 15, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 15, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 179f18b into kubernetes:master Jul 15, 2026
26 checks passed
@hakman
hakman deleted the release-notes-1.36-update branch July 16, 2026 05:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants