Sitelet https://github.com/kubernetes/kops/pull/18577
Skip to content

Add support for kube-apiserver --strict-transport-security-directives flag - #18577

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:kube-apiserver-hsts
Jul 12, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:kube-apiserver-hsts

Conversation

@hakman

@hakman hakman commented Jul 12, 2026 •

Copy link
Copy Markdown
Member

Refs #16478

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jul 12, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab July 12, 2026 05:21
@kubernetes-prow kubernetes-prow Bot added area/api size/S Denotes a PR that changes 10-29 lines, ignoring generated files. labels Jul 12, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 12, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rifelpet

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 12, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 19b2f39 into kubernetes:master Jul 12, 2026
21 of 26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants