Sitelet https://github.com/kubernetes/kops/pull/18563
Skip to content

dump: grant SSH access to Karpenter nodes via EC2 Instance Connect - #18563

Merged
kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
hakman:toolbox-dump-instance-connect
Jul 9, 2026
Merged

kubernetes-prow[bot] merged 2 commits into
kubernetes:masterfrom
hakman:toolbox-dump-instance-connect

Conversation

@hakman

@hakman hakman commented Jul 9, 2026

Copy link
Copy Markdown
Member

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 9, 2026 05:26
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 9, 2026
@hakman

hakman commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-scenario-aws-karpenter

@hakman
hakman force-pushed the toolbox-dump-instance-connect branch from 71b7c0c to 2fb3e96 Compare July 9, 2026 05:31
@hakman hakman changed the title tests/e2e: Grant SSH access to Karpenter nodes via EC2 Instance Connect dump: grant SSH access to Karpenter nodes via EC2 Instance Connect Jul 9, 2026
@hakman

hakman commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/test pull-kops-scenario-aws-karpenter

@ameukam

ameukam commented Jul 9, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 9, 2026
@hakman

hakman commented Jul 9, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 9, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit d778b94 into kubernetes:master Jul 9, 2026
29 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants