Sitelet https://github.com/kubernetes/kops/pull/18555
Skip to content

Automated cherry pick of #18450: etcd-manager: switch to go-runner-based distroless image - #18555

Merged
kubernetes-prow[bot] merged 3 commits into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18450-upstream-release-1.35
Jul 8, 2026
Merged

kubernetes-prow[bot] merged 3 commits into
kubernetes:release-1.35from
hakman:automated-cherry-pick-of-#18450-upstream-release-1.35

Conversation

@hakman

@hakman hakman commented Jul 8, 2026

Copy link
Copy Markdown
Member

Cherry pick of #18450 on release-1.35.

#18450: etcd-manager: switch to go-runner-based distroless image

For details on the cherry pick process, see the cherry pick requests page.

What type of PR is this?


@kubernetes-prow kubernetes-prow Bot added this to the v1.35 milestone Jul 8, 2026
@kubernetes-prow kubernetes-prow Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Jul 8, 2026
@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab July 8, 2026 15:24
@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jul 8, 2026
@hakman

hakman commented Jul 8, 2026

Copy link
Copy Markdown
Member Author

/cc @rifelpet @ameukam

@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 8, 2026 15:25
@ameukam

ameukam commented Jul 8, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 8, 2026
@hakman

hakman commented Jul 8, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 8, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit e4d1b65 into kubernetes:release-1.35 Jul 8, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants