Sitelet https://github.com/kubernetes/kops/pull/18541
Skip to content

Update cluster-autoscaler to v1.36.0 - #18541

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:update-cluster-autoscaler-1.36.0
Jul 5, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
hakman:update-cluster-autoscaler-1.36.0

Conversation

@hakman

@hakman hakman commented Jul 5, 2026

Copy link
Copy Markdown
Member

Signed-off-by: Ciprian Hacman <ciprian@hakman.dev>
@kubernetes-prow
kubernetes-prow Bot requested review from ameukam and rifelpet July 5, 2026 12:51
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jul 5, 2026
@hakman

hakman commented Jul 5, 2026

Copy link
Copy Markdown
Member Author

/approve

@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 5, 2026
@ameukam

ameukam commented Jul 5, 2026

Copy link
Copy Markdown
Member

/lgtm

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 5, 2026
kubernetes-prow Bot added a commit that referenced this pull request Jul 5, 2026
…-upstream-release-1.36

Automated cherry pick of #18541: Update cluster-autoscaler to v1.36.0
@kubernetes-prow
kubernetes-prow Bot merged commit 9fc2eec into kubernetes:master Jul 5, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants