Sitelet https://github.com/kubernetes/kops/pull/18526
Skip to content

fix: use JoinHostPost to tolerate IPv6 addresses - #18526

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
justinsb:use_joinhostport_for_ipv6
Jul 2, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
justinsb:use_joinhostport_for_ipv6

Conversation

@justinsb

@justinsb justinsb commented Jul 2, 2026

Copy link
Copy Markdown
Member

No description provided.

@kubernetes-prow
kubernetes-prow Bot requested review from olemarkus and zetaab July 2, 2026 20:29
@kubernetes-prow kubernetes-prow Bot added size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Jul 2, 2026
@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 2, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 2, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 96cb87c into kubernetes:master Jul 2, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants