Sitelet https://github.com/kubernetes/kops/pull/18519
Skip to content

Add an experimental roles feature flag. - #18519

Merged
kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
cheftako:multipleControlRoles
Jun 30, 2026
Merged

kubernetes-prow[bot] merged 1 commit into
kubernetes:masterfrom
cheftako:multipleControlRoles

Conversation

@cheftako

Copy link
Copy Markdown
Member

No new functionality yet.
Added 4 new role placholders, etcd, scheduler, ccm and kcm. Sets up the CLI API as well as the accessor functions.

No new functionality yet.
Added 4 new role placholders, etcd, scheduler, ccm and kcm.
Sets up the CLI API as well as the accessor functions.
@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jun 29, 2026
@cheftako

Copy link
Copy Markdown
Member Author

/assign @justinsb

@cheftako

Copy link
Copy Markdown
Member Author

/assign @hakman

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 30, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 30, 2026
@kubernetes-prow
kubernetes-prow Bot merged commit 568d0e0 into kubernetes:master Jun 30, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/api area/documentation cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants