Sitelet https://github.com/kubernetes/kops/pull/18498
Skip to content

Remove namespace from DO ClusterRole - #18498

Merged
k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rifelpet:do-ns
Jun 20, 2026
Merged

k8s-ci-robot merged 1 commit into
kubernetes:masterfrom
rifelpet:do-ns

Conversation

@rifelpet

Copy link
Copy Markdown
Member

DO jobs are failing because the channels container is failing to apply this manifest:

https://storage.googleapis.com/kubernetes-ci-logs/logs/e2e-kops-do-gossip/2068038730425307136/artifacts/cluster-info/kube-system/kops-channels-10.108.0.4/kops-channels.log

W0619 18:55:03.807097       1 results.go:63] error from apply on rbac.authorization.k8s.io/v1, Kind=ClusterRole kube-system/csi-do-node-driver-registrar-role: namespace "kube-system" was provided for cluster-scoped object rbac.authorization.k8s.io/v1, Kind=ClusterRole
I0619 18:55:04.007854       1 health.go:64] status conditions not found for StatefulSet.apps:kube-system/snapshot-controller
W0619 18:55:04.307159       1 results.go:56] consistency error (healthy counts): &applyset.ApplyResults{total:27, applySuccessCount:26, applyFailCount:1, healthyCount:26, unhealthyCount:0}
W0619 18:55:04.307204       1 apply_channel.go:119] error in apply iteration (will retry in 5s): applying "do://e2e-kops-space/e2e-e2e-kops-do-gossip.k8s.local/addons/bootstrap-channel.yaml": updating "digitalocean-csi-driver.addons.k8s.io": error updating addon from "do://e2e-kops-space/e2e-e2e-kops-do-gossip.k8s.local/addons/digitalocean-csi-driver.addons.k8s.io/k8s-1.22.yaml": error applying update: not all objects were applied; error applying update after prune: not all objects were applied

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jun 20, 2026
@k8s-ci-robot
k8s-ci-robot requested review from olemarkus and zetaab June 20, 2026 02:12
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hakman

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 20, 2026
k8s-ci-robot added a commit that referenced this pull request Jun 20, 2026
…98-origin-release-1.35

Automated cherry pick of #18498: Remove namespace from DO ClusterRole
@hakman

hakman commented Jun 20, 2026

Copy link
Copy Markdown
Member

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 20, 2026
@k8s-ci-robot
k8s-ci-robot merged commit c9b93c3 into kubernetes:master Jun 20, 2026
28 checks passed
k8s-ci-robot added a commit that referenced this pull request Jun 20, 2026
…98-origin-release-1.36

Automated cherry pick of #18498: Remove namespace from DO ClusterRole
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/addons cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants