Sitelet https://github.com/kubernetes/kops/pull/18496
Skip to content

Fixing LB behavior when you have both APIServer and Control Plane. - #18496

Merged
k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
cheftako:apiOnlyLB
Jun 19, 2026
Merged

k8s-ci-robot merged 3 commits into
kubernetes:masterfrom
cheftako:apiOnlyLB

Conversation

@cheftako

Copy link
Copy Markdown
Member

Initially the LB sent traffic to both.
The DNS None is a new case.
Now we only send traffic to the APIServer in this case.
This protects the Control Plane nodes to do core controller work.

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Jun 18, 2026
@k8s-ci-robot
k8s-ci-robot requested review from hakman and zetaab June 18, 2026 23:08
@k8s-ci-robot k8s-ci-robot added area/documentation area/provider/gcp Issues or PRs related to gcp provider labels Jun 18, 2026
Initially the LB sent traffic to both.
The DNS None is a new case.
Now we only send traffic to the APIServer in this case.
This protects the Control Plane nodes to do core controller work.
Remove separate tests.
Regenerated docs.
@k8s-ci-robot k8s-ci-robot added size/S Denotes a PR that changes 10-29 lines, ignoring generated files. and removed size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Jun 18, 2026
}

func (b *AutoscalingGroupModelBuilder) Build(c *fi.CloudupModelBuilderContext) error {
clusterHasApiServerOnly := false

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nit: clusterHasAPIServerOnlyInstanceGroups

@justinsb

Copy link
Copy Markdown
Member

Let's get a scenario test going so we can see this working, but .. lgtm. We probably should think about how to roll this out on other clouds also; we could consider creating a shared method like "ShouldAttachAPIServerLoadBalancer" or similar.

This is technically a behavioural change, but APIServerNodes is feature flagged and I think it's a good improvement so 👍

Name suggestion is not a blocker

/approve
/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jun 18, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: justinsb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 18, 2026
@k8s-ci-robot
k8s-ci-robot merged commit a84a7ac into kubernetes:master Jun 19, 2026
27 checks passed
@cheftako

Copy link
Copy Markdown
Member Author

This is technically a behavioural change, but APIServerNodes is feature flagged and I think it's a good improvement so 👍

It is a behavioral change. However it is limited to GCE use cases on as you say a feature flagged use case.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/documentation area/provider/gcp Issues or PRs related to gcp provider cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants